Skip to content

clusterloader2: bump k8s.io deps to v0.36.3 - #4207

Open
Jefftree wants to merge 2 commits into
kubernetes:masterfrom
Jefftree:bump-cl2-k8s-1.36
Open

Jefftree wants to merge 2 commits into
kubernetes:masterfrom
Jefftree:bump-cl2-k8s-1.36

Conversation

@Jefftree

@Jefftree Jefftree commented Jul 24, 2026

Copy link
Copy Markdown
Member

Bumps clusterloader2 k8s.io deps to v0.36.

I think we should turn on watchlist for the cl2 probe.

@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: Jefftree
Once this PR has been reviewed and has the lgtm label, please assign wojtek-t for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow kubernetes-prow Bot added the cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. label Jul 24, 2026
@kubernetes-prow
kubernetes-prow Bot requested review from mborsz and wojtek-t July 24, 2026 16:39
@kubernetes-prow kubernetes-prow Bot added the size/L Denotes a PR that changes 100-499 lines, ignoring generated files. label Jul 24, 2026
@Jefftree

Copy link
Copy Markdown
Member Author

/cc @serathius @p0lyn0mial

@kubernetes-prow

Copy link
Copy Markdown
Contributor

PR needs rebase.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@kubernetes-prow kubernetes-prow Bot added the needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. label Aug 7, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant