Skip to content

Move write throughput benchmark to 10KB pod - #4378

Open
serathius wants to merge 1 commit into
kubernetes:masterfrom
serathius:write-throughput-10KB
Open

serathius wants to merge 1 commit into
kubernetes:masterfrom
serathius:write-throughput-10KB

Conversation

@serathius

Copy link
Copy Markdown
Member

/kind feature
/cc @p0lyn0mial

@kubernetes-prow kubernetes-prow Bot added kind/feature Categorizes issue or PR as related to a new feature. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files. approved Indicates a PR has been approved by an approver from all required OWNERS files. labels Sep 18, 2026
@serathius
serathius force-pushed the write-throughput-10KB branch from 8934240 to 0f60c36 Compare September 18, 2026 10:50
@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: serathius

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow kubernetes-prow Bot added size/M Denotes a PR that changes 30-99 lines, ignoring generated files. and removed size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files. labels Sep 18, 2026
@serathius
serathius force-pushed the write-throughput-10KB branch from 0f60c36 to ecc770a Compare September 18, 2026 10:51
@serathius
serathius force-pushed the write-throughput-10KB branch from ecc770a to c9849f9 Compare September 18, 2026 10:52
@kubernetes-prow kubernetes-prow Bot added size/S Denotes a PR that changes 10-29 lines, ignoring generated files. and removed size/M Denotes a PR that changes 30-99 lines, ignoring generated files. labels Sep 18, 2026
@kubernetes-prow

Copy link
Copy Markdown
Contributor

@serathius: The following test failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
pull-perf-tests-ec2-master-scale-performance-100 c9849f9 link true /test pull-perf-tests-ec2-master-scale-performance-100

Full PR test history. Your PR dashboard. Please help us cut down on flakes by linking to an open issue when you hit one in your PR.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. kind/feature Categorizes issue or PR as related to a new feature. size/S Denotes a PR that changes 10-29 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant